fix: @graphiql/toolkit dependencies - #1968
Merged
Merged
Conversation
🦋 Changeset detectedLatest commit: 620e65b The changes in this PR will be included in the next version bump. This PR includes changesets to release 2 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
acao
force-pushed
the
fix/graphiql-toolkit-deps
branch
from
October 10, 2021 14:32
6f78692 to
620e65b
Compare
Codecov Report
@@ Coverage Diff @@
## main #1968 +/- ##
==========================================
+ Coverage 65.70% 65.89% +0.18%
==========================================
Files 85 86 +1
Lines 5106 5143 +37
Branches 1631 1640 +9
==========================================
+ Hits 3355 3389 +34
- Misses 1747 1750 +3
Partials 4 4
Continue to review full report at Codecov.
|
Merged
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Made the mistake of adding
optionalDependenciesto@graphiql/toolkit, some of which are vulnerable.This reverts that, and increments versions to the latest of graphql-ws, meros, and @n1ru4l/push-pull-async-iterable-iterator
Question: should these be
peerDependenciesand documentation updated to reflect this?